Skip to content

bump secretjs - #204

Merged
AustinWoetzel merged 4 commits into
mainfrom
develop
Sep 12, 2025
Merged

bump secretjs#204
AustinWoetzel merged 4 commits into
mainfrom
develop

Conversation

@AustinWoetzel

Copy link
Copy Markdown
Collaborator

No description provided.

@changeset-bot

changeset-bot Bot commented Sep 12, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: c8b0b52

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@shadeprotocol/shadejs Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@AustinWoetzel AustinWoetzel changed the title bump shadejs bump secretjs Sep 12, 2025
@netlify

netlify Bot commented Sep 12, 2025

Copy link
Copy Markdown

Deploy Preview for cool-empanada-69706a ready!

Name Link
🔨 Latest commit c8b0b52
🔍 Latest deploy log https://app.netlify.com/projects/cool-empanada-69706a/deploys/68c36db2d6c73a0008cc50be
😎 Deploy Preview https://deploy-preview-204--cool-empanada-69706a.netlify.app
📱 Preview on mobile
Toggle QR Code...

QR Code

Use your smartphone camera to open QR code link.

To edit notification comments on pull requests, go to your Netlify project configuration.

@AustinWoetzel
AustinWoetzel merged commit 5ac304f into main Sep 12, 2025
8 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant