Skip to content

Update secretjs dependency to version 1.21.0 - #203

Merged
AustinWoetzel merged 3 commits into
developfrom
AustinWoetzel-patch-1
Sep 12, 2025
Merged

Update secretjs dependency to version 1.21.0#203
AustinWoetzel merged 3 commits into
developfrom
AustinWoetzel-patch-1

Conversation

@AustinWoetzel

Copy link
Copy Markdown
Collaborator

No description provided.

@changeset-bot

changeset-bot Bot commented Sep 12, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 721f927

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@shadeprotocol/shadejs Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@AustinWoetzel
AustinWoetzel merged commit c8b0b52 into develop Sep 12, 2025
4 checks passed
@AustinWoetzel
AustinWoetzel deleted the AustinWoetzel-patch-1 branch September 12, 2025 00:47
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant