Skip to content

[ko] Translate content/ko/docs/setup/production-environment/tools/kub… - #57428

Open
llokr1 wants to merge 1 commit into
kubernetes:mainfrom
llokr1:ko-ha-etcd-with-kubeadm-setup
Open

[ko] Translate content/ko/docs/setup/production-environment/tools/kub…#57428
llokr1 wants to merge 1 commit into
kubernetes:mainfrom
llokr1:ko-ha-etcd-with-kubeadm-setup

Conversation

@llokr1

@llokr1 llokr1 commented Sep 5, 2026

Copy link
Copy Markdown
Contributor

…eadm/setup-ha-etcd-with-kubeadm.md

Description

This updates the Korean translation of content/ko/docs/setup/production-environment/tools/kubeadm/setup-ha-etcd-with-kubeadm.md

Issue

Closes: #57391

@kubernetes-prow
kubernetes-prow Bot requested review from Eundms and ianychoi September 5, 2026 06:39
@kubernetes-prow kubernetes-prow Bot added the area/localization General issues or PRs related to localization label Sep 5, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign wonyongg for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. language/ko Issues or PRs related to Korean language sig/docs Categorizes an issue or PR as relevant to SIG Docs. size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels Sep 5, 2026
@netlify

netlify Bot commented Sep 5, 2026

Copy link
Copy Markdown

Pull request preview available for checking

Built without sensitive environment variables

Name Link
🔨 Latest commit fab9931
🔍 Latest deploy log https://app.netlify.com/projects/kubernetes-io-main-staging/deploys/6a9bb920a256f200089a607e
😎 Deploy Preview https://deploy-preview-57428--kubernetes-io-main-staging.netlify.app
📱 Preview on mobile
Toggle QR Code...

QR Code

Use your smartphone camera to open QR code link.

To edit notification comments on pull requests, go to your Netlify project configuration.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

area/localization General issues or PRs related to localization cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. language/ko Issues or PRs related to Korean language sig/docs Categorizes an issue or PR as relevant to SIG Docs. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

[ko] Translate content/en/docs/setup/production-environment/tools/kubeadm/setup-ha-etcd-with-kubeadm.md into Korean

1 participant