Skip to content

Publish: CSI CBT beta blog - #57420

Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes:mainfrom
Amarachi-Ezeonyekwere:blog/schedule-article
Sep 7, 2026
Merged

Publish: CSI CBT beta blog#57420
kubernetes-prow[bot] merged 1 commit into
kubernetes:mainfrom
Amarachi-Ezeonyekwere:blog/schedule-article

Conversation

@Amarachi-Ezeonyekwere

Copy link
Copy Markdown
Contributor

Description

Schedules the CSI Changed Block Tracking blog post merged in #57401 for 2026-09-14 and removes draft: true

Issue

Schedules the CSI Changed Block Tracking blog post merged in #57401 for 2026-09-14 and removes draft: true

Closes: #57401

Signed-off-by: Amarachi Ezeonyekwere <203523295+Amarachi-Ezeonyekwere@users.noreply.github.com>
@linux-foundation-easycla

linux-foundation-easycla Bot commented Sep 4, 2026

Copy link
Copy Markdown

CLA Signed
The committers listed above are authorized under a signed CLA.

  • ✅ login: Amarachi-Ezeonyekwere / name: Amarachi Ezeonyekwere (a25236f)

@kubernetes-prow kubernetes-prow Bot added area/blog Issues or PRs related to the Kubernetes Blog subproject language/en Issues or PRs related to English language cncf-cla: no Indicates the PR's author has not signed the CNCF CLA. labels Sep 4, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

Welcome @Amarachi-Ezeonyekwere!

It looks like this is your first PR to kubernetes/website 🎉. Please refer to our pull request process documentation to help your PR have a smooth ride to approval.

You will be prompted by a bot to use commands during the review process. Do not be afraid to follow the prompts! It is okay to experiment. Here is the bot commands documentation.

You can also check if kubernetes/website has its own contribution guidelines.

You may want to refer to our testing guide if you run into trouble with your tests not passing.

If you are having difficulty getting your pull request seen, please follow the recommended escalation practices. Also, for tips and tricks in the contribution process you may want to read the Kubernetes contributor cheat sheet. We want to make sure your contribution gets all the attention it needs!

Thank you, and welcome to Kubernetes. 😃

@kubernetes-prow kubernetes-prow Bot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Sep 4, 2026
@netlify

netlify Bot commented Sep 4, 2026

Copy link
Copy Markdown

Pull request preview available for checking

Built without sensitive environment variables

Name Link
🔨 Latest commit a25236f
🔍 Latest deploy log https://app.netlify.com/projects/kubernetes-io-main-staging/deploys/6a9b32598d0373000861701a
😎 Deploy Preview https://deploy-preview-57420--kubernetes-io-main-staging.netlify.app
📱 Preview on mobile
Toggle QR Code...

QR Code

Use your smartphone camera to open QR code link.

To edit notification comments on pull requests, go to your Netlify project configuration.

@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. and removed cncf-cla: no Indicates the PR's author has not signed the CNCF CLA. labels Sep 4, 2026
@KashishV999

Copy link
Copy Markdown
Contributor

Thank you
/lgtm

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Sep 5, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

LGTM label has been added.

DetailsGit tree hash: da8435beb9b23f47f8b3cd730bfdedd6b71d1cf6

@KashishV999

Copy link
Copy Markdown
Contributor

@KashishV999

Copy link
Copy Markdown
Contributor

/retitle Publish: CSI CBT beta blog

@kubernetes-prow kubernetes-prow Bot changed the title blog: schedule CSI changed block tracking article for 2026-09-14 Publish: CSI CBT beta blog Sep 5, 2026

@Arhell Arhell left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Lgtm

@lmktfy

lmktfy commented Sep 7, 2026

Copy link
Copy Markdown
Member

/approve

@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: lmktfy

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 7, 2026
@kubernetes-prow
kubernetes-prow Bot merged commit 5803d7f into kubernetes:main Sep 7, 2026
6 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/blog Issues or PRs related to the Kubernetes Blog subproject cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. language/en Issues or PRs related to English language lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants