Skip to content

Fix title typo in Kubernetes objects documentation - #57411

Open
W0jno wants to merge 1 commit into
kubernetes:mainfrom
W0jno:patch-1
Open

Fix title typo in Kubernetes objects documentation#57411
W0jno wants to merge 1 commit into
kubernetes:mainfrom
W0jno:patch-1

Conversation

@W0jno

@W0jno W0jno commented Sep 3, 2026

Copy link
Copy Markdown

Description

"Objects" in polish language is "Obiekty", not "Objekty"

"Objects" in polish language is "Obiekty", not "Objekty"
@linux-foundation-easycla

Copy link
Copy Markdown

CLA Not Signed

@kubernetes-prow
kubernetes-prow Bot requested a review from mfilocha September 3, 2026 16:34
@kubernetes-prow kubernetes-prow Bot added the area/localization General issues or PRs related to localization label Sep 3, 2026
@kubernetes-prow
kubernetes-prow Bot requested a review from nvtkaszpir September 3, 2026 16:34
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign nvtkaszpir for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the language/pl Issues or PRs related to Polish language label Sep 3, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

Welcome @W0jno!

It looks like this is your first PR to kubernetes/website 🎉. Please refer to our pull request process documentation to help your PR have a smooth ride to approval.

You will be prompted by a bot to use commands during the review process. Do not be afraid to follow the prompts! It is okay to experiment. Here is the bot commands documentation.

You can also check if kubernetes/website has its own contribution guidelines.

You may want to refer to our testing guide if you run into trouble with your tests not passing.

If you are having difficulty getting your pull request seen, please follow the recommended escalation practices. Also, for tips and tricks in the contribution process you may want to read the Kubernetes contributor cheat sheet. We want to make sure your contribution gets all the attention it needs!

Thank you, and welcome to Kubernetes. 😃

@kubernetes-prow kubernetes-prow Bot added sig/docs Categorizes an issue or PR as relevant to SIG Docs. cncf-cla: no Indicates the PR's author has not signed the CNCF CLA. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Sep 3, 2026

@Arhell Arhell left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/easycla

@lmktfy

lmktfy commented Sep 7, 2026

Copy link
Copy Markdown
Member

Hello @W0jno. Can you see to the hyperlink that lets you license these changes to Kubernetes (Cloud Native Computing Foundation)?

Even though this is a tiny fix, we'd like a legal agreement that it is OK to include the fix into Kubernetes.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

area/localization General issues or PRs related to localization cncf-cla: no Indicates the PR's author has not signed the CNCF CLA. language/pl Issues or PRs related to Polish language sig/docs Categorizes an issue or PR as relevant to SIG Docs. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants