Skip to content

Add Product Engineering team as code owners - #114

Merged
Matt (matt-evervault) merged 1 commit into
mainfrom
feature/codeowners
Jul 9, 2026
Merged

Add Product Engineering team as code owners#114
Matt (matt-evervault) merged 1 commit into
mainfrom
feature/codeowners

Conversation

@matt-evervault

Copy link
Copy Markdown
Contributor

Updated CODEOWNERS file to include Product Engineering team.

Why

Add a short description about this PR.
Add links to issues, tech plans etc.

How

Describe how you've approached the problem

Updated CODEOWNERS file to include Product Engineering team.
@matt-evervault Matt (matt-evervault) self-assigned this Jul 8, 2026
@changeset-bot

changeset-bot Bot commented Jul 8, 2026

Copy link
Copy Markdown

⚠️ No Changeset found

Latest commit: bc748ac

Merging this PR will not cause a version bump for any packages. If these changes should not result in a new version, you're good to go. If these changes should result in a version bump, you need to add a changeset.

This PR includes no changesets

When changesets are added to this PR, you'll see the packages that this PR includes changesets for and the associated semver types

Click here to learn what changesets are, and how to add one.

Click here if you're a maintainer who wants to add a changeset to this PR

@matt-evervault
Matt (matt-evervault) requested a review from a team July 8, 2026 11:03
@matt-evervault
Matt (matt-evervault) merged commit 9c1ee56 into main Jul 9, 2026
5 checks passed
@matt-evervault
Matt (matt-evervault) deleted the feature/codeowners branch July 9, 2026 08:54
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Development

Successfully merging this pull request may close these issues.

2 participants